fix(napi): declare WASM runtime dependencies - #164
Open
Princesseuh wants to merge 2 commits into
Open
Princesseuh wants to merge 2 commits into
Princesseuh wants to merge 2 commits into
Conversation
🦋 Changeset detectedLatest commit: f42d916 The changes in this PR will be included in the next version bump. This PR includes changesets to release 3 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Princesseuh
commented
Sep 17, 2026
| "devDependencies": { | ||
| "@emnapi/runtime": "1.11.3", | ||
| "@napi-rs/cli": "3.5.1", | ||
| "@napi-rs/cli": "3.8.6", |
Member
Author
There was a problem hiding this comment.
New CLI version generates package.json for the WASM bindings with those deps properly there
ocavue
reviewed
Sep 17, 2026
| "devDependencies": { | ||
| "@emnapi/runtime": "1.11.3", | ||
| "@napi-rs/cli": "3.5.1", | ||
| "@napi-rs/cli": "3.8.6", |
Contributor
There was a problem hiding this comment.
I'm okay with bumping @napi-rs/cli to 3.8 in crates/astro_napi/package.json, but just a reminder that you probably don't want to bump @napi-rs/cli to 3.8 in crates/astro2tsx/package.json.
The latest version that works for both 2 packages is 3.7, as shown in #172
@napi-rs/cli 3.8+ in astro2tsx/package.json is blocking by issue napi-rs/napi-rs#3531
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Fixes #148